Repository navigation
Conversation
🦋 Changeset detectedLatest commit: 11f9545 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
rossiam
force-pushed
the
smartapps-commands
branch
from
October 1, 2026 14:51
3b68fda to
cd1df66
Compare
rossiam
force-pushed
the
smartapps-commands
branch
from
October 9, 2026 17:50
cd1df66 to
07de299
Compare
rossiam
force-pushed
the
smartapps-commands
branch
from
October 9, 2026 19:02
07de299 to
11f9545
Compare
dstanich
reviewed
Oct 9, 2026
| command => command.client.smartapps.list(smartAppListOptions), | ||
| ) | ||
|
|
||
| export const chooseSmartApp = chooseSmartAppFn() |
There was a problem hiding this comment.
AI detected nothing is importing this so it may be able to be removed.
dstanich
reviewed
Oct 9, 2026
| ['$0 smartapps', 'list all Smart Apps'], | ||
| [ | ||
| '$0 smartapps 1', | ||
| 'display details for the first smart app in the list retrieved by running "smartthings smartapps"', |
There was a problem hiding this comment.
Should capitalize Smart App since it is in other places.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Updated to use new smartapps endpoints for managing API Access apps.
smartappsreplacesappshierarchy.If the user has access to more than one organization, they will be prompted for which to use.
Improved error messages when input or output is non-interactive and user does not specify everything on the command line.
https://community.smartthings.com/t/a-new-enhanced-smartthings-api-experience/309947