Skip to content

openstack-k8s-operators dependency bump branch: main - #1188

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
openstack-dependency-bump/main
Sep 4, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
openstack-dependency-bump/main

Conversation

@github-actions

Copy link
Copy Markdown

Automated changes by create-pull-request GitHub action

@openshift-ci

openshift-ci Bot commented Aug 15, 2026

Copy link
Copy Markdown
Contributor

Hi @github-actions[bot]. Thanks for your PR.

I'm waiting for a openstack-k8s-operators member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1d8ed229c93b4b198ea9eb12ca9af399

✔️ openstack-meta-content-provider SUCCESS in 3h 49m 58s
✔️ nova-operator-kuttl SUCCESS in 1h 00m 53s
✔️ nova-operator-kuttl-placement SUCCESS in 59m 32s
❌ nova-operator-tempest-multinode FAILURE in 2h 45m 02s
❌ nova-operator-tempest-multinode-ceph FAILURE in 3h 15m 16s

@stuggi

stuggi commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

recheck

@github-actions
github-actions Bot force-pushed the openstack-dependency-bump/main branch from 79d2f5e to b4a4aa6 Compare August 29, 2026 05:08
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/799de06a1bae4ed3867620e34e7701df

✔️ openstack-meta-content-provider SUCCESS in 3h 11m 44s
✔️ nova-operator-kuttl SUCCESS in 1h 02m 22s
❌ nova-operator-kuttl-placement FAILURE in 1h 05m 29s
✔️ nova-operator-tempest-multinode SUCCESS in 2h 33m 27s
✔️ nova-operator-tempest-multinode-ceph SUCCESS in 2h 53m 56s
❌ nova-operator-cyborg-tempest-multinode RETRY_LIMIT in 19m 52s (non-voting)

@abays

abays commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

recheck

@abays abays left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm label Sep 4, 2026
@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abays, github-actions[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved label Sep 4, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit bba0bb2 into main Sep 4, 2026
7 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the openstack-dependency-bump/main branch September 4, 2026 19:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants