Added E2E for role ownership validation - #1209
openshift-merge-bot[bot] merged 3 commits into
Conversation
Signed-off-by: Alka Kumari <alkumari@redhat.com>
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
Signed-off-by: Alka Kumari <alkumari@redhat.com>
|
/test v4.21-ci-index-gitops-operator-bundle |
|
/lgtm |
|
/test v4.21-kuttl-sequential |
|
/test v4.21-kuttl-parallel |
1 similar comment
|
/test v4.21-kuttl-parallel |
|
/retest |
4 similar comments
|
/retest |
|
/retest |
|
/retest |
|
/retest |
|
/test v4.21-kuttl-sequential |
1 similar comment
|
/test v4.21-kuttl-sequential |
Signed-off-by: Alka Kumari <alkumari@redhat.com>
|
/test v4.21-kuttl-sequential |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: anandrkskd, svghadi The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
0266e44
into
redhat-developer:v1.21
What type of PR is this?
What does this PR do / why we need it:
Change git ref for argocd operator and adds E2E tests for role ownership validation
Have you updated the necessary documentation?
Which issue(s) this PR fixes:
Fixes #?
Test acceptance criteria:
How to test changes / Special notes to the reviewer: